Skip to content

Change from "extensions.v1beta1" to "networking.v1beta1" - #806

Merged
k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
bowei:remove-extensions
Jul 30, 2019
Merged

k8s-ci-robot merged 2 commits into
kubernetes:masterfrom
bowei:remove-extensions

Conversation

@bowei

@bowei bowei commented Jul 29, 2019

Copy link
Copy Markdown
Member
Change API group from "extensions" to "networking"

This migrates our code to use the networking.v1beta1.Ingress* API
types instead of extensions.v1beta1. This also fixes fixture YAMLs
to use "networking.k8s.io" as the fully-qualified resource kind.

bowei added 2 commits July 29, 2019 09:58
This migrates our code to use the networking.v1beta1.Ingress* API
types instead of extensions.v1beta1. This also fixes fixture YAMLs
to use "networking.k8s.io" as the fully-qualified resource kind.
@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files. labels Jul 29, 2019
@bowei

bowei commented Jul 29, 2019

Copy link
Copy Markdown
Member Author

/assign @rramkumar1 @freehan

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Jul 29, 2019
@bowei

bowei commented Jul 29, 2019

Copy link
Copy Markdown
Member Author

cc: @liggitt

@rramkumar1

Copy link
Copy Markdown
Contributor

@bowei We would need to update the YAML's in our GCP documentation as well.

[1] https://cloud.google.com/kubernetes-engine/docs/concepts/ingress

@bowei bowei changed the title Remove extensions Change from "extensions.v1beta1" to "networking.v1beta1" Jul 29, 2019
@freehan

freehan commented Jul 29, 2019

Copy link
Copy Markdown
Contributor

Will it still recognize ingress objects in extension api group?

I assume API machinery should have handled this kind of backward compatibility, right?

@bowei

bowei commented Jul 29, 2019

Copy link
Copy Markdown
Member Author

Yes, asked Jordan, this should work as long as it is used with v1.15+ k8s clusters

@freehan

freehan commented Jul 30, 2019

Copy link
Copy Markdown
Contributor

/lgtm
/approve

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Jul 30, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: bowei, freehan

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot
k8s-ci-robot merged commit cf2262e into kubernetes:master Jul 30, 2019
@bowei
bowei deleted the remove-extensions branch April 21, 2020 18:31
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. size/XL Denotes a PR that changes 500-999 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants