Skip to content

Implement custom request headers for backend config - #795

Merged
rramkumar1 merged 2 commits into
kubernetes:masterfrom
mcfedr:feature/customRequestHeaders
Aug 16, 2019
Merged

rramkumar1 merged 2 commits into
kubernetes:masterfrom
mcfedr:feature/customRequestHeaders

Conversation

@mcfedr

@mcfedr mcfedr commented Jul 9, 2019

Copy link
Copy Markdown
Contributor

I've updated #572

  • Add e2e test based on the CDN test
  • Config with a struct containing the headers list
  • Custom resources headers are now GA

@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Hi @mcfedr. Thanks for your PR.

I'm waiting for a kubernetes member to verify that this patch is reasonable to test. If it is, they should reply with /ok-to-test on its own line. Until that is done, I will not automatically test new commits in this PR, but the usual testing commands by org members will still work. Regular contributors should join the org to skip this step.

Once the patch is verified, the new status will be reflected by the ok-to-test label.

I understand the commands that are listed here.

Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository.

@k8s-ci-robot k8s-ci-robot added needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. size/L Denotes a PR that changes 100-499 lines, ignoring generated files. labels Jul 9, 2019
@k8s-ci-robot
k8s-ci-robot requested review from MrHohn and thockin July 9, 2019 20:06
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Thanks for your pull request. Before we can look at your pull request, you'll need to sign a Contributor License Agreement (CLA).

📝 Please follow instructions at https://git.k8s.io/community/CLA.md#the-contributor-license-agreement to sign the CLA.

It may take a couple minutes for the CLA signature to be fully registered; after that, please reply here with a new comment and we'll verify. Thanks.


Details

Instructions for interacting with me using PR comments are available here. If you have questions or suggestions related to my behavior, please file an issue against the kubernetes/test-infra repository. I understand the commands that are listed here.

@k8s-ci-robot k8s-ci-robot added the cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. label Jul 9, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

Welcome @mcfedr!

It looks like this is your first PR to kubernetes/ingress-gce 🎉. Please refer to our pull request process documentation to help your PR have a smooth ride to approval.

You will be prompted by a bot to use commands during the review process. Do not be afraid to follow the prompts! It is okay to experiment. Here is the bot commands documentation.

You can also check if kubernetes/ingress-gce has its own contribution guidelines.

You may want to refer to our testing guide if you run into trouble with your tests not passing.

If you are having difficulty getting your pull request seen, please follow the recommended escalation practices. Also, for tips and tricks in the contribution process you may want to read the Kubernetes contributor cheat sheet. We want to make sure your contribution gets all the attention it needs!

Thank you, and welcome to Kubernetes. 😃

@mcfedr

mcfedr commented Jul 9, 2019

Copy link
Copy Markdown
Contributor Author

cal should now be signed

@k8s-ci-robot k8s-ci-robot added cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. and removed cncf-cla: no Indicates the PR's author has not signed the CNCF CLA. labels Jul 9, 2019
@mcfedr

mcfedr commented Jul 9, 2019

Copy link
Copy Markdown
Contributor Author

@rramkumar1 as the previous PR by @pondohva seems to have stalled I've made the changes requested there, as well as rebasing on master

@mcfedr
mcfedr force-pushed the feature/customRequestHeaders branch from f524d0f to 47576dd Compare July 10, 2019 07:45
@rramkumar1

Copy link
Copy Markdown
Contributor

@mcfedr Thanks! We will take a look.

/ok-to-test

@k8s-ci-robot k8s-ci-robot added ok-to-test Indicates a non-member PR verified by an org member that is safe to test. and removed needs-ok-to-test Indicates a PR that requires an org member to verify it is safe to test. labels Jul 10, 2019
@k8s-ci-robot k8s-ci-robot added the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 1, 2019

@rramkumar1 rramkumar1 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Apologies for the delayed review, thanks for the work!

Generally LGTM, but you do need a rebase since a lot of changes have gone in before this.

Comment thread cmd/e2e-test/customrequestheaders_test.go Outdated
Comment thread cmd/e2e-test/customrequestheaders_test.go Outdated
Comment thread cmd/e2e-test/customrequestheaders_test.go Outdated
Comment thread pkg/backends/features/features.go Outdated
@mcfedr
mcfedr force-pushed the feature/customRequestHeaders branch from 47576dd to fd8e104 Compare August 7, 2019 10:50
@k8s-ci-robot k8s-ci-robot removed the needs-rebase Indicates a PR cannot be merged because it has merge conflicts with HEAD. label Aug 7, 2019
@mcfedr

mcfedr commented Aug 7, 2019

Copy link
Copy Markdown
Contributor Author

@rramkumar1 Addressed all issues and rebased.

@rramkumar1 rramkumar1 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

Thanks for the work! Please close out the related issue once this is merged.

Comment thread pkg/backends/features/customrequestheaders_test.go Outdated
Comment thread pkg/backends/features/features_test.go Outdated
@mcfedr
mcfedr force-pushed the feature/customRequestHeaders branch from fd8e104 to 46b9a59 Compare August 9, 2019 07:38
- Add e2e test
- Config with a struct containing the headers list
- Custom resources headers are now GA
@mcfedr
mcfedr force-pushed the feature/customRequestHeaders branch from 46b9a59 to 4564cee Compare August 13, 2019 08:34
@mcfedr

mcfedr commented Aug 16, 2019

Copy link
Copy Markdown
Contributor Author

@rramkumar1 Its good to go.

@rramkumar1 rramkumar1 left a comment

Copy link
Copy Markdown
Contributor

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

/lgtm

@k8s-ci-robot k8s-ci-robot added the lgtm "Looks good to me", indicates that a PR is ready to be merged. label Aug 16, 2019
@rramkumar1

Copy link
Copy Markdown
Contributor

/approve

@rramkumar1
rramkumar1 merged commit fc87c9c into kubernetes:master Aug 16, 2019
@k8s-ci-robot

Copy link
Copy Markdown
Contributor

[APPROVALNOTIFIER] This PR is APPROVED

This pull-request has been approved by: mcfedr, rramkumar1

The full list of commands accepted by this bot can be found here.

The pull request process is described here

Details Needs approval from an approver in each of these files:

Approvers can indicate their approval by writing /approve in a comment
Approvers can cancel approval by writing /approve cancel in a comment

@k8s-ci-robot k8s-ci-robot added the approved Indicates a PR has been approved by an approver from all required OWNERS files. label Aug 16, 2019
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

approved Indicates a PR has been approved by an approver from all required OWNERS files. cncf-cla: yes Indicates the PR's author has signed the CNCF CLA. lgtm "Looks good to me", indicates that a PR is ready to be merged. ok-to-test Indicates a non-member PR verified by an org member that is safe to test. size/L Denotes a PR that changes 100-499 lines, ignoring generated files.

Projects

None yet

Development

Successfully merging this pull request may close these issues.

4 participants